Skip to content

chore(deps) Bump github.com/go-git/go-billy/v5 to v5.9.0 to avoid CVE… - #72

Closed
nmirasch wants to merge 2 commits into
redhat-developer:masterfrom
nmirasch:CVE-2026-44740
Closed

nmirasch wants to merge 2 commits into
redhat-developer:masterfrom
nmirasch:CVE-2026-44740

Conversation

@nmirasch

Copy link
Copy Markdown

…-2026-44740
https://www.cve.org/CVERecord?id=CVE-2026-44740
GHSA-m3xc-h892-ggx6

Fixes: https://redhat.atlassian.net/browse/GITOPS-10734

What type of PR is this?

Uncomment only one /kind line, and delete the rest.
For example, > /kind bug would simply become: /kind bug

/kind bug
/kind cleanup
/kind failing-test
/kind enhancement
/kind documentation
/kind code-refactoring

What does this PR do / why we need it:

Have you updated the necessary documentation?

  • Documentation update is required by this PR.
  • Documentation has been updated.

Which issue(s) this PR fixes:

Fixes #?

How to test changes / Special notes to the reviewer:

@openshift-ci

openshift-ci Bot commented Sep 15, 2026

Copy link
Copy Markdown

[APPROVALNOTIFIER] This PR is NOT APPROVED

This pull-request has been approved by:
Once this PR has been reviewed and has the lgtm label, please assign trdoyle81 for approval. For more information see the Code Review Process.

The full list of commands accepted by this bot can be found here.

Details Needs approval from an approver in each of these files:

Approvers can indicate their approval by writing /approve in a comment
Approvers can cancel approval by writing /approve cancel in a comment

@openshift-ci

openshift-ci Bot commented Sep 15, 2026

Copy link
Copy Markdown

Hi @nmirasch. Thanks for your PR.

I'm waiting for a redhat-developer member to verify that this patch is reasonable to test. If it is, they should reply with /ok-to-test on its own line. Until that is done, I will not automatically test new commits in this PR, but the usual testing commands by org members will still work.

Tip

We noticed you've done this a few times! Consider joining the org to skip this step and gain /lgtm and other bot rights. We recommend asking approvers on your previous PRs to sponsor you.

Once the patch is verified, the new status will be reflected by the ok-to-test label.

I understand the commands that are listed here.

Details

Instructions for interacting with me using PR comments are available here. If you have questions or suggestions related to my behavior, please file an issue against the kubernetes-sigs/prow repository.

@nmirasch

Copy link
Copy Markdown
Author

/ok-to-test

@nmirasch

Copy link
Copy Markdown
Author

/retest

Signed-off-by: nmirasch <neus.miras@gmail.com>
Sign up for free to join this conversation on GitHub. Already have an account? Sign in to comment

Projects

None yet

Development

Successfully merging this pull request may close these issues.

1 participant