openstack-k8s-operators dependency bump branch: main - #2065
Conversation
|
Hi @github-actions[bot]. Thanks for your PR. I'm waiting for a openstack-k8s-operators member to verify that this patch is reasonable to test. If it is, they should reply with Regular contributors should join the org to skip this step. Once the patch is verified, the new status will be reflected by the I understand the commands that are listed here. DetailsInstructions for interacting with me using PR comments are available here. If you have questions or suggestions related to my behavior, please file an issue against the kubernetes-sigs/prow repository. |
|
Merge Failed. This change or one of its cross-repo dependencies was unable to be automatically merged with the current state of its repository. Please rebase the change and upload a new patchset. |
/test openstack-operator-build-deploy-kuttl-4-20 |
|
recheck |
|
Build failed (check pipeline). Post ❌ openstack-k8s-operators-content-provider NODE_FAILURE Node(set) request 100-0000193207 failed in 0s |
|
recheck |
|
Build failed (check pipeline). Post ❌ openstack-k8s-operators-content-provider NODE_FAILURE Node(set) request 100-0000193431 failed in 0s |
|
recheck |
/test openstack-operator-build-deploy-kuttl-4-20 |
|
Build failed (check pipeline). Post ❌ openstack-k8s-operators-content-provider NODE_FAILURE Node(set) request 100-0000193469 failed in 0s |
|
recheck |
|
/test openstack-operator-build-deploy-kuttl-4-20 |
|
Build failed (check pipeline). Post ✔️ openstack-k8s-operators-content-provider SUCCESS in 15m 20s |
|
recheck |
|
Build failed (check pipeline). Post ✔️ openstack-k8s-operators-content-provider SUCCESS in 14m 14s |
|
recheck vexxhost maintenance over |
|
/test precommit-check |
|
[APPROVALNOTIFIER] This PR is APPROVED This pull-request has been approved by: danpawlik, github-actions[bot], karelyatin, stuggi The full list of commands accepted by this bot can be found here. The pull request process is described here DetailsNeeds approval from an approver in each of these files:
Approvers can indicate their approval by writing |
[1] including neutron-operator fixes is merged, so let's enable neutron and agents. [1] openstack-k8s-operators/openstack-operator#2065 Related-Issue: #OSPRH-28744 Related-Issue: #OSPRH-33113 Signed-off-by: Yatin Karel <ykarel@redhat.com>
Automated changes by create-pull-request GitHub action