build(deps): bump the aws-sdk group with 3 updates - #1001
Conversation
Bumps the aws-sdk group with 3 updates: [github.com/aws/aws-sdk-go-v2/config](https://github.com/aws/aws-sdk-go-v2), [github.com/aws/aws-sdk-go-v2/credentials](https://github.com/aws/aws-sdk-go-v2) and [github.com/aws/aws-sdk-go-v2/service/sts](https://github.com/aws/aws-sdk-go-v2). Updates `github.com/aws/aws-sdk-go-v2/config` from 1.33.4 to 1.33.5 - [Release notes](https://github.com/aws/aws-sdk-go-v2/releases) - [Commits](aws/aws-sdk-go-v2@config/v1.33.4...config/v1.33.5) Updates `github.com/aws/aws-sdk-go-v2/credentials` from 1.20.4 to 1.20.5 - [Release notes](https://github.com/aws/aws-sdk-go-v2/releases) - [Commits](aws/aws-sdk-go-v2@service/mq/v1.20.4...service/mq/v1.20.5) Updates `github.com/aws/aws-sdk-go-v2/service/sts` from 1.50.0 to 1.51.0 - [Release notes](https://github.com/aws/aws-sdk-go-v2/releases) - [Commits](aws/aws-sdk-go-v2@service/s3/v1.50.0...service/s3/v1.51.0) --- updated-dependencies: - dependency-name: github.com/aws/aws-sdk-go-v2/config dependency-version: 1.33.5 dependency-type: direct:production update-type: version-update:semver-patch dependency-group: aws-sdk - dependency-name: github.com/aws/aws-sdk-go-v2/credentials dependency-version: 1.20.5 dependency-type: direct:production update-type: version-update:semver-patch dependency-group: aws-sdk - dependency-name: github.com/aws/aws-sdk-go-v2/service/sts dependency-version: 1.51.0 dependency-type: direct:production update-type: version-update:semver-minor dependency-group: aws-sdk ... Signed-off-by: dependabot[bot] <support@github.com>
|
Important Review skippedIgnore keyword(s) in the title. ⛔ Ignored keywords (3)
Please check the settings in the CodeRabbit UI or the ⚙️ Run configurationConfiguration used: Repository: openshift/backplane-cli/.coderabbit.yaml Review profile: CHILL Plan: Enterprise Run ID: You can disable this status message by setting the Use the checkbox below for a quick retry:
Comment |
|
Hi @dependabot[bot]. Thanks for your PR. I'm waiting for a openshift member to verify that this patch is reasonable to test. If it is, they should reply with Regular contributors should join the org to skip this step. Once the patch is verified, the new status will be reflected by the I understand the commands that are listed here. DetailsInstructions for interacting with me using PR comments are available here. If you have questions or suggestions related to my behavior, please file an issue against the kubernetes-sigs/prow repository. |
|
[APPROVALNOTIFIER] This PR is NOT APPROVED This pull-request has been approved by: dependabot[bot] The full list of commands accepted by this bot can be found here. DetailsNeeds approval from an approver in each of these files:Approvers can indicate their approval by writing |
Codecov Report✅ All modified and coverable lines are covered by tests. Additional details and impacted files@@ Coverage Diff @@
## main #1001 +/- ##
==========================================
- Coverage 55.67% 55.64% -0.04%
==========================================
Files 80 80
Lines 6398 6398
==========================================
- Hits 3562 3560 -2
- Misses 2401 2402 +1
- Partials 435 436 +1 🚀 New features to boost your workflow:
|
Bumps the aws-sdk group with 3 updates: github.com/aws/aws-sdk-go-v2/config, github.com/aws/aws-sdk-go-v2/credentials and github.com/aws/aws-sdk-go-v2/service/sts.
Updates
github.com/aws/aws-sdk-go-v2/configfrom 1.33.4 to 1.33.5Commits
4e0240aRelease 2026-09-14c2d42dfRegenerated Clients1548c19Update endpoints model9fab903Update API modela7eaa10Release 2026-09-11c6f53e8Regenerated Clientse80f4faUpdate endpoints model55b26d7Update API model113bc91Release 2026-09-100b5dc66Regenerated ClientsUpdates
github.com/aws/aws-sdk-go-v2/credentialsfrom 1.20.4 to 1.20.5Commits
f2706c8Release 2023-12-081dac0c9Regenerated Clientscbd9216Update API modelc7357bbfix: reinstate presence of retryer when functional opts run but still respect...c5c34b3fix: translation of ini service sections into shared config (#2416)b3c7fbfupdate express cache key (#2414)9b90af4fix: add non-vhostable buckets to path when using legacy endpoint resolver (#...Updates
github.com/aws/aws-sdk-go-v2/service/stsfrom 1.50.0 to 1.51.0Commits
e9b00e2Release 2024-02-220cfc53eRegenerated Clients92d6c19Update API model18adb31add middleware snapshot tests (#2513)7888f0eRelease 2024-02-214a9cd1dRegenerated Clientsbebcd8cUpdate SDK's smithy-go dependency to v1.20.1374e0b9Update API model98b0dc8dep: drop dependency on go-cmp in protocol tests (#2506)de4f646Release 2024-02-20Dependabot will resolve any conflicts with this PR as long as you don't alter it yourself. You can also trigger a rebase manually by commenting
@dependabot rebase.Dependabot commands and options
You can trigger Dependabot actions by commenting on this PR:
@dependabot rebasewill rebase this PR@dependabot recreatewill recreate this PR, overwriting any edits that have been made to it@dependabot show <dependency name> ignore conditionswill show all of the ignore conditions of the specified dependency@dependabot ignore <dependency name> major versionwill close this group update PR and stop Dependabot creating any more for the specific dependency's major version (unless you unignore this specific dependency's major version or upgrade to it yourself)@dependabot ignore <dependency name> minor versionwill close this group update PR and stop Dependabot creating any more for the specific dependency's minor version (unless you unignore this specific dependency's minor version or upgrade to it yourself)@dependabot ignore <dependency name>will close this group update PR and stop Dependabot creating any more for the specific dependency (unless you unignore this specific dependency or upgrade to it yourself)@dependabot unignore <dependency name>will remove all of the ignore conditions of the specified dependency@dependabot unignore <dependency name> <ignore condition>will remove the ignore condition of the specified dependency and ignore conditions