Skip to content

chore(deps): bump cloud.google.com/go/compute/metadata from 0.9.0 to 0.9.1 - #1356

Open
dependabot[bot] wants to merge 1 commit into
masterfrom
dependabot/go_modules/cloud.google.com/go/compute/metadata-0.9.1
Open

dependabot[bot] wants to merge 1 commit into
masterfrom
dependabot/go_modules/cloud.google.com/go/compute/metadata-0.9.1

Conversation

@dependabot

@dependabot dependabot Bot commented on behalf of github Sep 21, 2026

Copy link
Copy Markdown
Contributor

Bumps cloud.google.com/go/compute/metadata from 0.9.0 to 0.9.1.

Release notes

Sourced from cloud.google.com/go/compute/metadata's releases.

compute/metadata: v0.9.1

0.9.1 (2026-09-17)

Bug Fixes

  • compute/metadata: Timeout initial metadata subscriber GET request (#20488) (f06f6b3), refs #20487
Changelog

Sourced from cloud.google.com/go/compute/metadata's changelog.

Changes

0.123.0 (2025-09-18)

Features

  • internal/stategen: Populate the latest googleapis commit (#12880) (7b017a0)
  • librariangen: Implement the build command (#12817) (14734c8)

Bug Fixes

  • internal/librariangen: Add link to source commit in release notes (#12881) (1c06cc6)
  • internal/librariangen: Fix CHANGES.md headers (#12849) (baf515d)
  • internal/librariangen: Remove go mod init/tidy from postprocessor (#12832) (1fe506a)
  • internal/librariangen: Test for error path with flags (#12830) (f0da7b2)
  • internal/postprocessor: Add dlp to skip-module-scan-paths (#12857) (45a7d9b)
  • librariangen: Honor original container contract (#12846) (71c8fd3)
  • librariangen: Improvements to release-init (#12842) (0db677a)
  • stategen: Specify an appropriate tag format for google-cloud-go (#12835) (ffcff33)

0.122.0 (2025-09-04)

Features

  • internal/librariangen: Add release-init command (#12751) (52e84cc)

Bug Fixes

  • internal/godocfx: Better support for v2 modules (#12797) (4bc8785)
  • internal/godocfx: Module detection when tidy errors (#12801) (83d46cd)
  • internal/librariangen: Fix goimports errors (#12765) (83bdaa4)

0.121.6 (2025-08-14)

Bug Fixes

  • internal/librariangen: Fix Dockerfile permissions for go mod tidy (#12704) (0e70a0b)

0.121.5 (2025-08-12)

Bug Fixes

... (truncated)

Commits
  • 66a482b chore(main): release auth 0.9.1 (#10735)
  • d8887df feat(datastore): reference new protos (#10724)
  • 5e67869 fix(auth): Setting expireEarly to default when the value is 0 (#10732)
  • 6f026e3 chore(main): release vertexai 0.13.0 (#10472)
  • 0ee1430 feat(vertexai): update tokenizer documentation and pull new code (#10718)
  • 9afb797 chore(main): release datastore 1.18.0 (#10468)
  • 946a5fc feat(gkeconnect/gateway): removed the nonfunctional GatewayService and replac...
  • 7785cad fix(spanner): wait for things to complete (#10095)
  • 1c80ee7 chore(spanner): drop errors backwards compatibility (#9757)
  • 7ef0ded fix(spanner): ensure defers run at the right time in tests (#9759)
  • Additional commits viewable in compare view

Dependabot compatibility score

Dependabot will resolve any conflicts with this PR as long as you don't alter it yourself. You can also trigger a rebase manually by commenting @dependabot rebase.


Dependabot commands and options

You can trigger Dependabot actions by commenting on this PR:

  • @dependabot rebase will rebase this PR
  • @dependabot recreate will recreate this PR, overwriting any edits that have been made to it
  • @dependabot show <dependency name> ignore conditions will show all of the ignore conditions of the specified dependency
  • @dependabot ignore this major version will close this PR and stop Dependabot creating any more for this major version (unless you reopen the PR or upgrade to it yourself)
  • @dependabot ignore this minor version will close this PR and stop Dependabot creating any more for this minor version (unless you reopen the PR or upgrade to it yourself)
  • @dependabot ignore this dependency will close this PR and stop Dependabot creating any more for this dependency (unless you reopen the PR or upgrade to it yourself)

Bumps [cloud.google.com/go/compute/metadata](https://github.com/googleapis/google-cloud-go) from 0.9.0 to 0.9.1.
- [Release notes](https://github.com/googleapis/google-cloud-go/releases)
- [Changelog](https://github.com/googleapis/google-cloud-go/blob/main/CHANGES.md)
- [Commits](googleapis/google-cloud-go@v0.9.0...auth/v0.9.1)

---
updated-dependencies:
- dependency-name: cloud.google.com/go/compute/metadata
  dependency-version: 0.9.1
  dependency-type: direct:production
  update-type: version-update:semver-patch
...

Signed-off-by: dependabot[bot] <support@github.com>
@dependabot dependabot Bot added the ok-to-test Indicates a non-member PR verified by an org member that is safe to test. label Sep 21, 2026
@kubernetes-prow

Copy link
Copy Markdown
Contributor

Hi @dependabot[bot]. Thanks for your PR.

I'm waiting for a kubernetes member to verify that this patch is reasonable to test. If it is, they should reply with /ok-to-test on its own line. Until that is done, I will not automatically test new commits in this PR, but the usual testing commands by org members will still work.

Regular contributors should join the org to skip this step.

Once the patch is verified, the new status will be reflected by the ok-to-test label.

I understand the commands that are listed here.

Details

Instructions for interacting with me using PR comments are available here. If you have questions or suggestions related to my behavior, please file an issue against the kubernetes-sigs/prow repository.

@kubernetes-prow kubernetes-prow Bot added cncf-cla: yes Indicates the PR's author has signed the CNCF CLA. size/M Denotes a PR that changes 30-99 lines, ignoring generated files. labels Sep 21, 2026
@kubernetes-prow

Copy link
Copy Markdown
Contributor

[APPROVALNOTIFIER] This PR is NOT APPROVED

This pull-request has been approved by: dependabot[bot]
Once this PR has been reviewed and has the lgtm label, please assign sjenning for approval. For more information see the Code Review Process.

The full list of commands accepted by this bot can be found here.

Details Needs approval from an approver in each of these files:

Approvers can indicate their approval by writing /approve in a comment
Approvers can cancel approval by writing /approve cancel in a comment

@kubernetes-prow

Copy link
Copy Markdown
Contributor

@dependabot[bot]: The following tests failed, say /retest to rerun all failed tests or /retest-required to rerun all mandatory failed tests:

Test name Commit Details Required Rerun command
pull-npd-verify-gomod d2bed17 link true /test pull-npd-verify-gomod
pull-npd-test d2bed17 link true /test pull-npd-test
pull-npd-e2e-test d2bed17 link true /test pull-npd-e2e-test
pull-npd-build d2bed17 link true /test pull-npd-build

Full PR test history. Your PR dashboard. Please help us cut down on flakes by linking to an open issue when you hit one in your PR.

Details

Instructions for interacting with me using PR comments are available here. If you have questions or suggestions related to my behavior, please file an issue against the kubernetes-sigs/prow repository. I understand the commands that are listed here.

Sign up for free to join this conversation on GitHub. Already have an account? Sign in to comment

Labels

cncf-cla: yes Indicates the PR's author has signed the CNCF CLA. ok-to-test Indicates a non-member PR verified by an org member that is safe to test. size/M Denotes a PR that changes 30-99 lines, ignoring generated files.

Projects

None yet

Development

Successfully merging this pull request may close these issues.

0 participants