Conversation
ref: LOG-10108
|
/hold |
|
[APPROVALNOTIFIER] This PR is APPROVED This pull-request has been approved by: jcantrill The full list of commands accepted by this bot can be found here. The pull request process is described here DetailsNeeds approval from an approver in each of these files:
Approvers can indicate their approval by writing |
|
Important Review skippedAuto reviews are disabled on base/target branches other than the default branch. Please check the settings in the CodeRabbit UI or the ⚙️ Run configurationConfiguration used: Organization UI Review profile: CHILL Plan: Advanced Run ID: You can disable this status message by setting the Use the checkbox below for a quick retry:
Comment |
|
/lgtm |
|
continue to hold until 6.2 release |
|
blocked by openshift/release#84947 |
|
/test images |
|
retesting to confirm |
|
Continue to hold awaiting the release of 6.2.13 /label verified Upstream release verified base image is ubi-micro |
|
/hold cancel |
|
/retest |
|
@jcantrill: The following tests failed, say
Full PR test history. Your PR dashboard. DetailsInstructions for interacting with me using PR comments are available here. If you have questions or suggestions related to my behavior, please file an issue against the kubernetes-sigs/prow repository. I understand the commands that are listed here. |
ref: LOG-10108
https://redhat.atlassian.net/browse/LOG-10108
cc @vparfonov @Clee2691