Skip to content

Add Dependabot and CODEOWNERS files - #32

Open
jgarber623-cargosense wants to merge 2 commits into
add-actions-workflowsfrom
add-dependabot-and-codeowners
Open

Add Dependabot and CODEOWNERS files#32
jgarber623-cargosense wants to merge 2 commits into
add-actions-workflowsfrom
add-dependabot-and-codeowners

Conversation

@jgarber623-cargosense

Copy link
Copy Markdown

That's it.

@jgarber623-cargosense
jgarber623-cargosense force-pushed the add-dependabot-and-codeowners branch from 396b8cb to 604dd10 Compare September 10, 2026 12:19
Sign up for free to join this conversation on GitHub. Already have an account? Sign in to comment

Labels

None yet

Projects

None yet

Development

Successfully merging this pull request may close these issues.

1 participant